Skip to content

Make gc logging more verbose - #4375

Open
mjudeikis wants to merge 1 commit into
kcp-dev:mainfrom
mjudeikis:e2e-shard-vmodule
Open

mjudeikis wants to merge 1 commit into
kcp-dev:mainfrom
mjudeikis:e2e-shard-vmodule

Conversation

@mjudeikis

Copy link
Copy Markdown
Contributor

Summary

Flakes in the GC controller are hard to debug in CI. Making the module more verbose so we have more logs.

What Type of PR Is This?

/kind flake

Related Issue(s)

Fixes #

Release Notes

NONE

@kcp-ci-bot kcp-ci-bot added release-note-none Denotes a PR that doesn't merit a release note. kind/flake Categorizes issue or PR as related to a flaky test. dco-signoff: yes Indicates the PR's author has signed the DCO. labels Sep 17, 2026
@kcp-ci-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign embik for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kcp-ci-bot kcp-ci-bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 17, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates the PR's author has signed the DCO. kind/flake Categorizes issue or PR as related to a flaky test. release-note-none Denotes a PR that doesn't merit a release note. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants